Skip to content

Updating CODEOWNERS to @aws-neuron/neuron-sdk - #210

Merged
cdbhatia-aws merged 1 commit into
mainfrom
migrate-codeowners
Sep 22, 2026
Merged

cdbhatia-aws merged 1 commit into
mainfrom
migrate-codeowners

Conversation

@cdbhatia-aws

Copy link
Copy Markdown
Contributor

Issue #, if available:

Updating CODEOWNERS to @aws-neuron/neuron-sdk

@cdbhatia-aws
cdbhatia-aws requested a review from a team as a code owner September 22, 2026 20:41
@cdbhatia-aws
cdbhatia-aws merged commit 15c5fd1 into main Sep 22, 2026
6 checks passed
@cdbhatia-aws
cdbhatia-aws deleted the migrate-codeowners branch September 22, 2026 20:50
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants