Skip to content

CODEOWNERS: Remove community team ownership of readme & issue template - #2944

Open
wjt wants to merge 1 commit into
mainfrom
wjt/codeowners-remove-community-team-ownership-of-readme-issue-template
Open

wjt wants to merge 1 commit into
mainfrom
wjt/codeowners-remove-community-team-ownership-of-readme-issue-template

Conversation

@wjt

@wjt wjt commented Sep 21, 2026

Copy link
Copy Markdown
Member

The community function is now part of the games team!

The community function is now part of the games team!
@wjt
wjt requested a review from a team as a code owner September 21, 2026 16:31
@wjt
wjt requested a review from lasr21 September 21, 2026 16:32
@github-actions

Copy link
Copy Markdown

Play this branch at https://play.threadbare.game/branches/endlessm/wjt/codeowners-remove-community-team-ownership-of-readme-issue-template/.

(This launches the game from the start, not directly at the change(s) in this pull request.)

@wjt
wjt added this pull request to stack #2947 September 21, 2026 17:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant