Skip to content

CODEOWNERS: Make new Threadbare Maintainers group the main owner - #2946

Merged
manuq merged 1 commit into
mainfrom
wjt/codeowners-make-new-threadbare-maintainers-group-the-main-owner
Sep 22, 2026
Merged

manuq merged 1 commit into
mainfrom
wjt/codeowners-make-new-threadbare-maintainers-group-the-main-owner

Conversation

@wjt

@wjt wjt commented Sep 21, 2026

Copy link
Copy Markdown
Member

Previously this was the Games team at Endless Access, but:

  1. Not every member of that team is a code contributor - yet they were
    still being tagged for review on every pull request. (As of today,
    this is the majority of that team.)

  2. In principle, someone could be a Threadbare maintainer without being
    an Endless Access employee & member of the Games team.

I have created a threadbare-maintainers team. Make this the main
codeowner for the game, in place of the Games team.

(The Games team still has the ability to merge pull requests to main
without review, as does this new team, but in both cases the change must
go through a PR and a special checkbox must be checked.)

@wjt
wjt changed the base branch from main to wjt/codeowners-remove-community-team-ownership-of-readme-issue-template September 21, 2026 17:01
@wjt
wjt added this pull request to stack #2947 September 21, 2026 17:01
@github-actions

Copy link
Copy Markdown

Play this branch at https://play.threadbare.game/branches/endlessm/wjt/codeowners-make-new-threadbare-maintainers-group-the-main-owner/.

(This launches the game from the start, not directly at the change(s) in this pull request.)

Base automatically changed from wjt/codeowners-remove-community-team-ownership-of-readme-issue-template to main September 22, 2026 11:41
Previously this was the Games team at Endless Access, but:

1. Not every member of that team is a code contributor - yet they were
   still being tagged for review on every pull request. (As of today,
   this is the majority of that team.)

2. In principle, someone could be a Threadbare maintainer without being
   an Endless Access employee & member of the Games team.

I have created a threadbare-maintainers team. Make this the main
codeowner for the game, in place of the Games team.

(The Games team still has the ability to merge pull requests to `main`
without review, as does this new team, but in both cases the change must
go through a PR and a special checkbox must be checked.)
@manuq
manuq force-pushed the wjt/codeowners-make-new-threadbare-maintainers-group-the-main-owner branch from d44146b to fef8f12 Compare September 22, 2026 11:41
@wjt
wjt marked this pull request as ready for review September 22, 2026 12:03
@wjt
wjt requested a review from a team as a code owner September 22, 2026 12:03
@manuq
manuq merged commit 5a0f4e1 into main Sep 22, 2026
7 checks passed
@manuq
manuq deleted the wjt/codeowners-make-new-threadbare-maintainers-group-the-main-owner branch September 22, 2026 12:15
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants