Skip to content

feat: bump to PostgREST v14.18 - #2446

Merged
laurenceisla merged 1 commit into
developfrom
laurence/pgrst-14.18
Sep 12, 2026
Merged

feat: bump to PostgREST v14.18#2446
laurenceisla merged 1 commit into
developfrom
laurence/pgrst-14.18

Conversation

@laurenceisla

Copy link
Copy Markdown
Contributor

What kind of change does this PR introduce?

Bumps PostgREST from v14.17 to v14.18

What is the current behavior?

v14.17 did not fix the "JWT issued at future" issue.

What is the new behavior?

We tested the fix included in v14.18 on affected projects and the problem didn't reappear.

@laurenceisla
laurenceisla requested review from a team as code owners September 11, 2026 18:59
@laurenceisla

Copy link
Copy Markdown
Contributor Author

Smoke tests pending, will update once they're done.

@laurenceisla
laurenceisla force-pushed the laurence/pgrst-14.18 branch 2 times, most recently from 71a7f66 to ae81bb6 Compare September 12, 2026 00:57
@laurenceisla

Copy link
Copy Markdown
Contributor Author

All smoke tests are passing (Sep. 11):

https://github.com/supabase/supadev/actions?query=actor%3Alaurenceisla

The test-postgres-engines-with-smoke is known to be failing and is skipped in CUFs.

@laurenceisla
laurenceisla added this pull request to the merge queue Sep 12, 2026
Merged via the queue into develop with commit 18464f5 Sep 12, 2026
45 checks passed
@laurenceisla
laurenceisla deleted the laurence/pgrst-14.18 branch September 12, 2026 02:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants