Skip to content

Use github-native CODEOWNERS - #6709

Merged
janniklasrose merged 3 commits into
mainfrom
janniklasrose/maintainer-approval-blocking
Sep 22, 2026
Merged

janniklasrose merged 3 commits into
mainfrom
janniklasrose/maintainer-approval-blocking

Conversation

@janniklasrose

@janniklasrose janniklasrose commented Sep 16, 2026

Copy link
Copy Markdown
Member

Changes

  1. Remove the maintainer-approval workflow
  2. Consolidate .github/OWNERS & .github/OWNERTEAMS into .github/CODEOWNERS
  3. To model "CLI maintainers can approve everything", include that team at the start of every entry
  4. Linter to make sure 3. is always true

Why

maintainer-approval check run has shown to be unreliable (requiring rebase despite getting sufficient approvals)

Tests

GitHub does not have a CODEOWNERS playground :-(

Comment thread .github/CODEOWNERS
@@ -0,0 +1,84 @@
# Maintainers can approve any PR. Keep them on every rule because GitHub uses

@janniklasrose janniklasrose Sep 16, 2026

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

File equivalent to old OWNERS but team: replaced with @databricks/.

Created:

  • databricks/eng-deco-cli
  • databricks/eng-deco-dabs
  • databricks/eng-deco-ax
  • databricks/eng-deco-platform

@eng-dev-ecosystem-bot

eng-dev-ecosystem-bot commented Sep 16, 2026

Copy link
Copy Markdown
Collaborator

Integration test report

Commit: 18f5626

Run: 35751091623

Env 🔄​flaky ✅​pass 🙈​skip Time
✅​ aws linux 276 15 5:26
✅​ aws windows 278 13 3:48
🔄​ azure linux 2 273 15 5:21
✅​ azure windows 277 13 4:02
✅​ gcp linux 276 15 5:56
✅​ gcp windows 278 13 4:14
Test Name azure linux
🔄​ TestFsRmNonEmptyDirectory 🔄​f
🔄​ TestFsRmNonEmptyDirectory/uc-volumes 🔄​f
Top 3 slowest tests (at least 2 minutes):
duration env testname
4:00 azure windows TestAccept
3:59 gcp windows TestAccept
3:46 aws windows TestAccept

Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERTEAMS
Comment thread .github/OWNERS
@janniklasrose
janniklasrose marked this pull request as ready for review September 16, 2026 14:07
…ner-approval-blocking

# Conflicts:
#	.github/OWNERS
@janniklasrose
janniklasrose added this pull request to the merge queue Sep 22, 2026
Merged via the queue into main with commit 78c0305 Sep 22, 2026
41 checks passed
@janniklasrose
janniklasrose deleted the janniklasrose/maintainer-approval-blocking branch September 22, 2026 16:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants